Skip to content

chore(deps): bump github.com/tektoncd/pipeline from 1.15.2 to 1.15.3 - #2010

Open
dependabot[bot] wants to merge 2 commits into
release-v0.29.xfrom
dependabot/go_modules/release-v0.29.x/github.com/tektoncd/pipeline-1.15.3
Open

dependabot[bot] wants to merge 2 commits into
release-v0.29.xfrom
dependabot/go_modules/release-v0.29.x/github.com/tektoncd/pipeline-1.15.3

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 25, 2026 •

Copy link
Copy Markdown
Contributor

Bumps github.com/tektoncd/pipeline from 1.15.2 to 1.15.3.

Release notes

Sourced from github.com/tektoncd/pipeline's releases.

Tekton Pipeline release v1.15.3 "Toyger Orisa" LTS

Installation one-liner

kubectl apply -f https://infra.tekton.dev/tekton-releases/pipeline/previous/v1.15.3/release.yaml

Attestation

The Rekor UUID for this release is 108e9186e8c5677a1cca30d273b8c9dacc3ecea843087a7743386bf7355ef7a283afd3ed1829b921

Obtain the attestation:

REKOR_UUID=108e9186e8c5677a1cca30d273b8c9dacc3ecea843087a7743386bf7355ef7a283afd3ed1829b921
rekor-cli get --uuid $REKOR_UUID --format json | jq -r .Attestation | jq .

Verify that all container images in the attestation are in the release file:

RELEASE_FILE=https://infra.tekton.dev/tekton-releases/pipeline/previous/v1.15.3/release.yaml
REKOR_UUID=108e9186e8c5677a1cca30d273b8c9dacc3ecea843087a7743386bf7355ef7a283afd3ed1829b921
Obtains the list of images with sha from the attestation
REKOR_ATTESTATION_IMAGES=$(rekor-cli get --uuid "$REKOR_UUID" --format json | jq -r .Attestation | jq -r '.subject[]|.name + ":v1.15.3@sha256:" + .digest.sha256')
Download the release file
curl -L "$RELEASE_FILE" > release.yaml
For each image in the attestation, match it to the release file
for image in $REKOR_ATTESTATION_IMAGES; do
printf $image; grep -q $image release.yaml && echo " ===> ok" || echo " ===> no match";
done

Changes

Features

... (truncated)

Commits
  • 64924d4 [release-v1.15.x] bump github.com/sigstore/sigstore
  • 379e5a0 [release-v1.15.x] bump the all group in /tekton with 4 updates
  • f1a9876 [release-v1.15.x] bump agilepathway/label-checker from 1.6.65 to 1.6.66
  • c75d6df [release-v1.15.x] bump the sigstore group with 2 updates
  • 8d1a538 [release-v1.15.x] bump the all group in /tekton with 6 updates
  • 1e6bc8c [release-v1.15.x] bump zizmorcore/zizmor-action from 0.6.3 to 0.6.4
  • 30b5b86 [release-v1.15.x] bump github.com/sigstore/sigstore/pkg/signature/kms/azure
  • 72a1cc3 [release-v1.15.x] bump the all group in /tekton with 4 updates
  • 40b0d3d [release-v1.15.x] bump github.com/sigstore/sigstore
  • c4ac71a [release-v1.15.x] bump chainguard-dev/actions/setup-kind
  • Additional commits viewable in compare view

@dependabot dependabot Bot added dependencies Used by dependabot - identifies all PRs created by dependabot kind/misc Categorizes issue or PR as a miscellaneuous one. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note-none Denotes a PR that doesnt merit a release note. labels Sep 25, 2026
@tekton-robot tekton-robot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Sep 25, 2026
@dependabot
dependabot Bot force-pushed the dependabot/go_modules/release-v0.29.x/github.com/tektoncd/pipeline-1.15.3 branch from 8ae7c00 to cba5e74 Compare September 29, 2026 06:49
infernus01
infernus01 previously approved these changes Sep 29, 2026

@infernus01 infernus01 left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@tekton-robot tekton-robot added the lgtm Indicates that a PR is ready to be merged. label Sep 29, 2026
@tekton-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: infernus01

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@tekton-robot tekton-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 29, 2026
Bumps [github.com/tektoncd/pipeline](https://github.com/tektoncd/pipeline) from 1.15.2 to 1.15.3.
- [Release notes](https://github.com/tektoncd/pipeline/releases)
- [Changelog](https://github.com/tektoncd/pipeline/blob/main/releases.md)
- [Commits](tektoncd/pipeline@v1.15.2...v1.15.3)

---
updated-dependencies:
- dependency-name: github.com/tektoncd/pipeline
  dependency-version: 1.15.3
  dependency-type: direct:production
  update-type: version-update:semver-patch
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/go_modules/release-v0.29.x/github.com/tektoncd/pipeline-1.15.3 branch from cba5e74 to 75085d6 Compare September 29, 2026 06:52
@tekton-robot tekton-robot removed the lgtm Indicates that a PR is ready to be merged. label Sep 29, 2026
@tekton-robot

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

Signed-off-by: Jawed khelil <jkhelil@redhat.com>
@jkhelil
jkhelil force-pushed the dependabot/go_modules/release-v0.29.x/github.com/tektoncd/pipeline-1.15.3 branch from 1fdab01 to b3a77cb Compare September 29, 2026 11:02
@jkhelil jkhelil mentioned this pull request Sep 30, 2026
5 of 6 tasks
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. dependencies Used by dependabot - identifies all PRs created by dependabot kind/misc Categorizes issue or PR as a miscellaneuous one. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note-none Denotes a PR that doesnt merit a release note. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants