Skip to content

Stop insisting on chart version bump on stable branches also - #3217

Merged
kubernetes-prow[bot] merged 2 commits into
kubernetes:masterfrom
shiftstack:chart-lint
Sep 14, 2026
Merged

kubernetes-prow[bot] merged 2 commits into
kubernetes:masterfrom
shiftstack:chart-lint

Conversation

@stephenfin

Copy link
Copy Markdown
Member

What this PR does / why we need it:

In PR #3195 we modified the chart release process so that we no longer required a bump in the chart version on master. Unfortunately, we still have an issue on stable release-* branches. Anyone that backports a fix using the /cherry-pick slash command will see a failing chart lint job on said stable branch. This can only be fixed by incrementing the version field, but no one (including the maintainers) has the ability to push to the k8s-infra-cherrypick-robot repo fork that the bot uses. This means the only practical way to backport fixes that include chart changes is to do so manually, which is a lot more work for very little gain.

Rather than doing this, drop the version check entirely. This means we will have to manually bump the chart version periodically, but that's not a huge ask and is very similar to what we already do for CPO itself. To make this easier, we make the hack/bump-version.sh smarter so that it can start automatically bumping chart versions.

While here, we also improve the release procedure doc further, clarifying the purpose of different tags and the instances where you should bump different fields.

Which issue this PR fixes(if applicable):

Ref #3194

Special notes for reviewers:

(none)

Release note:

NONE

@kubernetes-prow kubernetes-prow Bot added the release-note-none Denotes a PR that doesn't merit a release note. label Sep 11, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from kayrus and zetaab September 11, 2026 16:34
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Sep 11, 2026
@stephenfin

Copy link
Copy Markdown
Member Author

/assign @mandre

Comment thread docs/release-procedure.md Outdated
In PR kubernetes#3195 we modified the chart release process so that we no longer
required a bump in the chart version on `master`. Unfortunately, we
still have an issue on stable `release-*` branches. Anyone that
backports a fix using the `/cherry-pick` slash command will see a
failing chart lint job on said stable branch. This can only be fixed by
incrementing the `version` field, but no one (including the maintainers)
has the ability to push to the `k8s-infra-cherrypick-robot` repo fork
that the bot uses. This means the only practical way to backport fixes
that include chart changes is to do so manually, which is a lot more
work for very little gain.

Rather than doing this, drop the version check entirely. This means we
will have to manually bump the chart version periodically, but that's
not a huge ask and is very similar to what we already do for CPO itself.
To make this easier, we make the `hack/bump-version.sh` smarter so that
it can start automatically bumping chart versions.

Signed-off-by: Stephen Finucane <stephenfin@redhat.com>
Encourage users to run the master version of the release script (it
doesn't exist on older branches). Also explain the meaning of the 3
chart-only tags.

Signed-off-by: Stephen Finucane <stephenfin@redhat.com>

@mandre mandre left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Sep 14, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: mandre

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 14, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 047d99b into kubernetes:master Sep 14, 2026
7 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesn't merit a release note. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants